fakechroot: apply patch to fix crash in __readlinkat_chk - #269911
Merged
Merged
Conversation
Contributor
Author
|
cc @kirillrdy |
kirillrdy
reviewed
Nov 25, 2023
Member
There was a problem hiding this comment.
Suggested change
| sha256 = "sha256-wMIZ3hW5XkRXQYBMADlN6kxhDSiEr84PGWBW+f4b4Ko="; | |
| hash = "sha256-wMIZ3hW5XkRXQYBMADlN6kxhDSiEr84PGWBW+f4b4Ko="; |
yorickvP
force-pushed
the
fix-readlinkat
branch
from
November 26, 2023 14:31
e603197 to
e446bff
Compare
kirillrdy
approved these changes
Nov 26, 2023
kirillrdy
left a comment
Member
There was a problem hiding this comment.
@yorickvP for some reason now ofborg-eval fails, although nixpkgs-review worked 🤷
but I can confirm that it fixes crash, reproducible via https://discourse.nixos.org/t/cant-get-coredump-from-nixbld-user/31090
Member
|
@ofborg eval |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description of changes
#246134
dex4er/fakechroot#114
dex4er/fakechroot#115
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Priorities
Add a 👍 reaction to pull requests you find important.