Conversation
baloo
marked this pull request as draft
November 2, 2024 04:01
nix-owners
Bot
requested review from
Artturin,
Ericson2314,
RossComputerGuy,
alyssais,
emilazy,
infinisil,
philiptaron and
reckenrode
November 2, 2024 04:02
Contributor
|
I'd love to review and merge this, @baloo. What's blocking it right now, other than the merge conflict in |
Member
Author
|
the uefi build does not work (see #353050) for a reason I'm not able to figure out. If you have any knowledge of the inner workings of clang/llvm, would love your help there |
With the new gnu-config support, we add UEFI system triples and system doubles in nixpkgs.
…/5495945c851a7bc9770eda5a3ed63c25 This commit fixes the CI failures with PR 231951: - https://gist.github.com/GrahamcOfBorg/5495945c851a7bc9770eda5a3ed63c25
baloo
force-pushed
the
baloo/lanzaboote-uefi-stub
branch
from
December 2, 2024 18:57
2dbd2f2 to
ebe0c59
Compare
Member
Author
|
Cleaned up the merge conflicts. |
Member
@NixOS/llvm or just @RossComputerGuy |
13 tasks
13 tasks
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This is a rebase of the work done in #231951
This is split between two PRs:
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.