chore(deps): update dependency svelte-awesome-color-picker to v4 - #4909
Open
renovate[bot] wants to merge 3 commits into
Open
renovate[bot] wants to merge 3 commits into
renovate[bot] wants to merge 3 commits into
Conversation
🦋 Changeset detectedLatest commit: 0d8c694 The changes in this PR will be included in the next version bump. This PR includes changesets to release 3 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 1, 2025 02:53
e7450ad to
29b7673
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 1, 2025 16:58
29b7673 to
2070c4a
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 3, 2025 14:17
2070c4a to
13ecdc7
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 3, 2025 19:24
13ecdc7 to
97c377c
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 3, 2025 23:34
97c377c to
f979e28
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 4, 2025 21:23
f979e28 to
d05b5f9
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 10, 2025 23:30
ebab54a to
86aae9a
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 11, 2025 19:26
86aae9a to
3d7292c
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 15, 2025 02:45
3d7292c to
003487e
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 15, 2025 23:07
003487e to
af6c0a6
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 16, 2025 12:47
af6c0a6 to
cacfc5e
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 16, 2025 12:51
cacfc5e to
982861e
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 16, 2025 12:57
982861e to
a3c6178
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 16, 2025 13:07
a3c6178 to
e420cf7
Compare
renovate
Bot
force-pushed
the
renovate/svelte-awesome-color-picker-4.x
branch
from
April 17, 2025 06:19
e420cf7 to
e8aef83
Compare
Contributor
JinIgarashi
temporarily deployed
to
Svelte UNDP Components
April 17, 2025 08:28 — with
GitHub Actions
Inactive
JinIgarashi
temporarily deployed
to
Svelte UNDP Design
April 17, 2025 08:28 — with
GitHub Actions
Inactive
Contributor
Author
Edited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
JinIgarashi
temporarily deployed
to
Svelte UNDP Design
May 13, 2025 12:55 — with
GitHub Actions
Inactive
JinIgarashi
temporarily deployed
to
Svelte UNDP Components
May 13, 2025 12:55 — with
GitHub Actions
Inactive
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

This PR contains the following updates:
^3.1.4->^4.0.0Release Notes
Ennoriel/svelte-awesome-color-picker (svelte-awesome-color-picker)
v4.0.1Patch Changes
aeaa655: Fix "change to" mode not using custom texts - thanks to Snazzahv4.0.0Major Changes
35c887c: set the position props default value to responsive instead of fixedd384e21: revert picker indicator cursor position extrapolation to bound to the wrapper picker element (issue #76)c275511: migrate codebase to svelte 5Patch Changes
de2725f: fix alpha slider background in case of a portal wrapper (fix issue #93)0fd681f: up deps1c8f4d1: support rtl alignment (issue #90)Configuration
📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR was generated by Mend Renovate. View the repository job log.