Skip to content

fix: prevent double-binding of ClusterLease to ClusterInstance - #38

Merged
caxu-rh merged 1 commit into
mainfrom
prevent-lease-double-bind
Sep 16, 2026
Merged

caxu-rh merged 1 commit into
mainfrom
prevent-lease-double-bind

Conversation

@caxu-rh

@caxu-rh caxu-rh commented Sep 16, 2026

Copy link
Copy Markdown
Collaborator

Closes #26.

Signed-off-by: Caleb Xu <caxu@redhat.com>
@exe-prow-github-app

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: caxu-rh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@exe-prow-github-app exe-prow-github-app Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 16, 2026
@caxu-rh
caxu-rh merged commit ecf1156 into main Sep 16, 2026
7 of 8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Prevent double-binding of ClusterLease to one ClusterInstance

1 participant