Skip to content

NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 - Follow-up - #203

Open
openshift-monitoring-bot[bot] wants to merge 1 commit into
openshift:mainfrom
rhobs:automated-updates-main
Open

NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 - Follow-up#203
openshift-monitoring-bot[bot] wants to merge 1 commit into
openshift:mainfrom
rhobs:automated-updates-main

Conversation

@openshift-monitoring-bot

@openshift-monitoring-bot openshift-monitoring-bot Bot commented Sep 3, 2026

Copy link
Copy Markdown

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.

@openshift-merge-bot

Copy link
Copy Markdown

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci
openshift-ci Bot requested review from jan--f and slashpai September 3, 2026 10:14

@simonpasquier simonpasquier left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm
/verified by @simonpasquier

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 4, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@simonpasquier: This PR has been marked as verified by @simonpasquier.

Details

In response to this:

/approve
/lgtm
/verified by @simonpasquier

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 4, 2026
@openshift-merge-bot

Copy link
Copy Markdown

Scheduling required tests:
/test e2e-agnostic-cmo
/test e2e-aws
/test e2e-aws-upgrade

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 4, 2026
Signed-off-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 9, 2026
@machine424

Copy link
Copy Markdown

/retitle [bot] Bump openshift/thanos to v0.42.4 - Follow-up
/lgtm

@openshift-ci openshift-ci Bot changed the title [bot] Bump openshift/thanos to v0.42.4 [bot] Bump openshift/thanos to v0.42.4 - Follow-up Sep 10, 2026
@machine424

Copy link
Copy Markdown

/retitle NO-JIRA: [bot] Bump openshift/thanos to v0.42.4

@openshift-ci openshift-ci Bot changed the title [bot] Bump openshift/thanos to v0.42.4 - Follow-up NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 Sep 10, 2026
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 10, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@machine424

Copy link
Copy Markdown

/retitle NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 - Follow-up

@openshift-ci openshift-ci Bot changed the title NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 NO-JIRA: [bot] Bump openshift/thanos to v0.42.4 - Follow-up Sep 10, 2026
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 10, 2026
@openshift-merge-bot

Copy link
Copy Markdown

Scheduling required tests:
/test e2e-agnostic-cmo
/test e2e-aws
/test e2e-aws-upgrade

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: machine424, openshift-monitoring-bot[bot], simonpasquier

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [machine424,simonpasquier]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 6caa7c9 and 2 for PR HEAD ea7804f in total

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-upgrade ea7804f link true /test e2e-aws-upgrade
ci/prow/e2e-agnostic-cmo ea7804f link true /test e2e-agnostic-cmo
ci/prow/e2e-aws ea7804f link true /test e2e-aws

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants